Skip to content

just a test - #2245

Closed
Mephistic wants to merge 1 commit into
codeforboston:mainfrom
Mephistic:typesense-test
Closed

Mephistic wants to merge 1 commit into
codeforboston:mainfrom
Mephistic:typesense-test

Conversation

@Mephistic

Copy link
Copy Markdown
Collaborator

Just testing the Typesense query for the Legislator Profile sidebar - don't merge

@vercel

vercel Bot commented Sep 15, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated
maple-dev Ready Ready Preview Sep 16, 2026 12:03am UTC
1 Skipped Deployment
Project Deployment Actions Updated
maple-prod Ignored Ignored Sep 16, 2026 12:03am UTC

Request Review

@Mephistic

Copy link
Copy Markdown
Collaborator Author

The real PR this was a test run for has been merged, so this is fine to close.

@Mephistic Mephistic closed this Sep 22, 2026

This branch was successfully deployed

1 active deployment
Preview – maple-dev — f6d1c634 Deployed Sep 16, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant