Skip to content

pkg/utils: add ELN as supported distro - #1814

Merged
debarshiray merged 1 commit into
containers:mainfrom
yselkowitz:eln
Sep 8, 2026
Merged

pkg/utils: add ELN as supported distro#1814
debarshiray merged 1 commit into
containers:mainfrom
yselkowitz:eln

Conversation

@yselkowitz

Copy link
Copy Markdown
Contributor

No description provided.

@gemini-code-assist gemini-code-assist Bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Code Review

This pull request adds support for Fedora ELN to Toolbx by introducing ELN-specific utility functions, registering the "eln" distribution, and updating the documentation and tests. The reviewer noted that the host detection logic in getHostID() must be updated to check for VARIANT_ID == "eln" when ID == "fedora" to ensure Fedora ELN hosts are correctly identified and default to ELN containers instead of standard Fedora containers.

Important

The consumer version of Gemini Code Assist on GitHub is being sunset. Starting June 18, 2026, new organization installations will be blocked, and all code review activity will officially cease on July 17, 2026.
For more details on the timeline and next steps, please review the Help Documentation.

Comment thread src/pkg/utils/utils.go
@yselkowitz

Copy link
Copy Markdown
Contributor Author

ping @debarshiray

@debarshiray debarshiray left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Interesting idea!

If Fedora ELN has ID=eln in os-release(5), then it seems natural to add it as a separate distro. Although, I hope we don't run into any namespace problems with other downstream distributors by not having a fedora prefix.

However, does this really work? From reading the changes, it seems that it uses the quay.io/fedora/eln:latest image. If I got that right, then:

$ toolbox create --image quay.io/fedora/eln:latest
Created container: eln-latest
Enter with: toolbox enter eln-latest
$ toolbox enter eln-latest 
Error: failed to modify user rishi with UID 1000: usermod(1) not found

Poking at the image with podman run confirms the problem:

$ podman run --interactive --rm --tty quay.io/fedora/eln:latest /bin/bash
[root@bb0c5c79fdcc /]# useradd
bash: useradd: command not found
[root@bb0c5c79fdcc /]# usermod
bash: usermod: command not found
[root@bb0c5c79fdcc /]# passwd
bash: passwd: command not found

In terms of having a similar user experience as the default host shell on Fedora or Red Hat Enterprise Linux Workstation, I see:

$ podman run --interactive --rm --tty quay.io/fedora/eln:latest /bin/bash
[root@e8a5adf33a34 /]# 
[root@e8a5adf33a34 /]# which
bash: which: command not found
[root@e8a5adf33a34 /]# man
bash: man: command not found

Here is a complete list of image requirements for Toolbx with some context and explanation.

I wonder if it will be possible to add an eln-toolbox image, just like we have the fedora-toolbox and the UBI-based toolbox images. The reason for the separate Toolbx images is to reliably deal with problems like these without getting into a tug-of-war with the default base images for the distribution, which have different goals.

In this particular case, just like ELN is (Red Hat) Enterprise Linux Next, the eln-toolbox image can help with the onboarding of the toolbox image for the upcoming RHEL.

@debarshiray

Copy link
Copy Markdown
Member

Build failed. https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/ec85aa64270a4c6faf391eeb10b3ab4c

unit-test RETRY_LIMIT in 29s

[...]

unit-test-restricted RETRY_LIMIT in 28s
system-test-fedora-rawhide-commands-options RETRY_LIMIT in 30s
system-test-fedora-rawhide-runtime-environment-arch-fedora RETRY_LIMIT in 27s
system-test-fedora-rawhide-runtime-environment-ubuntu RETRY_LIMIT in 33s

These CI failures should be solved by rebasing against main, specifically commit 93f882c04cbc5a81.

@yselkowitz

Copy link
Copy Markdown
Contributor Author

We already have a toolbox image, quay.io/fedora/eln-toolbox, and this code uses it.

@debarshiray

Copy link
Copy Markdown
Member

We already have a toolbox image, quay.io/fedora/eln-toolbox, and this code uses it.

That's fantastic!

My apologies for misreading the code and assuming that this pull request uses quay.io/fedora/eln:latest. I forgot and missed that the image's basename is mentioned in Distro.ImageBasename in src/pkg/utils/utils.go and not in src/pkg/utils/<distro>.go.

Could you please point me to the sources of the quay.io/fedora/eln-toolbox image for reference? I am curious if the sources are a Container/Dockerfile or KIWI descriptions or something else. I couldn't find it in fedora-kiwi-descriptions, where the fedora-toolbox image is defined - neither as a branch nor in teams/cloud/container.xml.

We try to have a copy of the image sources in toolbox.git/images. For the Arch Linux and Ubuntu images, these are the canonical sources and they are built and published by the CI, because those distributions don't have their own image building and publishing infrastructure. For the Fedora and RHEL images, they are copies to make it easier for those not familiar with the distributions to see how the images look like and make changes if changes to the toolbox(1) code require any. Unfortunately, we haven't updated the copies for Fedora after the move away from Container/Dockerfiles due to lack of time to extract the relevant KIWI description subset from fedora-kiwi-descriptions.

@yselkowitz

yselkowitz commented Aug 25, 2026

Copy link
Copy Markdown
Contributor Author

@debarshiray debarshiray left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This looks very good to me! Just a few bureaucratic and nitpicky details:

Would you be willing to add yourself to .github/CODEOWNERS for the src/pkg/utils/eln.go file? You clearly know more about Fedora ELN than me. If you are unavailable, then I will be the fallback, but having you there will formally clarify who the subject matter expert is.

It will be good to add some system tests for the Fedora ELN image to test/system. The tests are tagged to make it possible run subsets of them. eg., simultaneously on different CI hosts to reduce time. It's done using a mixture of Bats' tags and a hacky TOOLBX_TEST_SYSTEM_TAGS environment variable to work around some missing feature in Bats.

The are some deficiencies in how the system tests handle multiple distributions and releases, but it shouldn't matter for a rolling-release distribution like ELN and you can mimic what's done for Arch Linux. eg., you have to add the name of the ELN Toolbx image to the IMAGES associative array or hash table in test/system/libs/helpers.bash, do the _pull_and_cache_distro_image dance in test/system/setup_suite.bash, and then sprinkle the tests across the test/system/*.bats files.

Some of the tests look a bit silly, but it seems worth it because of the growing host and container distro matrix. We might as well automate everything at the risk of repeating some things for the sake of mental peace and reassurance. :)

Nitpick: could you please add a link to this pull request to the Git commit message?

Comment thread doc/toolbox.1.md
Comment thread src/pkg/utils/utils_test.go
Comment thread src/pkg/utils/eln.go Outdated
@debarshiray

Copy link
Copy Markdown
Member

https://forge.fedoraproject.org/releng/kiwi-descriptions/src/branch/rawhide/teams/eln/container.xml

Aha, cool! I should have noticed the eln subdirectory there. Sorry about that.

@debarshiray

Copy link
Copy Markdown
Member

Build failed. https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/ec85aa64270a4c6faf391eeb10b3ab4c
unit-test RETRY_LIMIT in 29s
[...]
unit-test-restricted RETRY_LIMIT in 28s
system-test-fedora-rawhide-commands-options RETRY_LIMIT in 30s
system-test-fedora-rawhide-runtime-environment-arch-fedora RETRY_LIMIT in 27s
system-test-fedora-rawhide-runtime-environment-ubuntu RETRY_LIMIT in 33s

These CI failures should be solved by rebasing against main, specifically commit 93f882c04cbc5a81.

Could you please rebase against main? It would be good to see the CI turn green.

yselkowitz added a commit to yselkowitz/toolbox that referenced this pull request Aug 27, 2026
containers#1814

Signed-off-by: Yaakov Selkowitz <yselkowi@redhat.com>
@yselkowitz

Copy link
Copy Markdown
Contributor Author

Rebased and incorporated your suggestions. Haven't looked at test/system yet.

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/c81a538e4f0b47239af525f545e823df

✔️ unit-test SUCCESS in 2m 16s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 12s
✔️ unit-test-restricted SUCCESS in 3m 13s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 45m 24s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 30m 41s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 7m 26s
✔️ system-test-fedora-44-commands-options SUCCESS in 43m 04s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 31m 31s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 8m 25s
✔️ system-test-fedora-43-commands-options SUCCESS in 42m 04s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 31m 40s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 7m 47s
✔️ system-test-fedora-42-commands-options SUCCESS in 37m 25s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 25m 17s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 6m 43s
system-test-fedora-coreos-next-commands-options NODE_FAILURE Node(set) request 200-0000184451 failed in 0s
system-test-fedora-coreos-next-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000184452 failed in 0s
system-test-fedora-coreos-next-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000184453 failed in 0s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 42m 05s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 33m 54s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 9m 15s
system-test-fedora-coreos-testing-commands-options NODE_FAILURE Node(set) request 200-0000184457 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000184458 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000184459 failed in 0s

@debarshiray

Copy link
Copy Markdown
Member

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/e3219230f9f94af98d0ddb9a90a6ef08

✔️ unit-test SUCCESS in 2m 17s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 12s
✔️ unit-test-restricted SUCCESS in 2m 17s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 44m 03s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 33m 10s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 7m 16s
✔️ system-test-fedora-44-commands-options SUCCESS in 41m 27s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 32m 46s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 7m 30s
✔️ system-test-fedora-43-commands-options SUCCESS in 38m 04s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 26m 39s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 7m 47s
✔️ system-test-fedora-42-commands-options SUCCESS in 33m 02s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 28m 45s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 6m 03s
system-test-fedora-coreos-next-commands-options NODE_FAILURE Node(set) request 200-0000184789 failed in 0s
system-test-fedora-coreos-next-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000184790 failed in 0s
system-test-fedora-coreos-next-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000184791 failed in 0s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 37m 23s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 35m 08s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 8m 35s
system-test-fedora-coreos-testing-commands-options NODE_FAILURE Node(set) request 200-0000184795 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000184796 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000184797 failed in 0s

@debarshiray

Copy link
Copy Markdown
Member

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/8eace5541b6643cd82f54f91f6a79ca4

✔️ unit-test SUCCESS in 2m 11s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 10s
✔️ unit-test-restricted SUCCESS in 2m 12s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 45m 37s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 29m 36s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 7m 04s
✔️ system-test-fedora-44-commands-options SUCCESS in 43m 24s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 32m 45s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 6m 57s
✔️ system-test-fedora-43-commands-options SUCCESS in 37m 39s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 33m 26s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 7m 43s
✔️ system-test-fedora-42-commands-options SUCCESS in 32m 00s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 22m 31s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 6m 07s
system-test-fedora-coreos-next-commands-options NODE_FAILURE Node(set) request 200-0000184848 failed in 0s
system-test-fedora-coreos-next-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000184849 failed in 0s
system-test-fedora-coreos-next-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000184850 failed in 0s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 43m 22s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 34m 28s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 8m 31s
system-test-fedora-coreos-testing-commands-options NODE_FAILURE Node(set) request 200-0000184854 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000184855 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000184856 failed in 0s

@debarshiray

Copy link
Copy Markdown
Member

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/9fecbcaa95954cc9824fcdd43ecbecdd

✔️ unit-test SUCCESS in 2m 14s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 11s
✔️ unit-test-restricted SUCCESS in 2m 16s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 43m 41s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 32m 14s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 7m 20s
✔️ system-test-fedora-44-commands-options SUCCESS in 38m 44s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 28m 40s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 6m 51s
✔️ system-test-fedora-43-commands-options SUCCESS in 35m 41s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 30m 47s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 7m 14s
✔️ system-test-fedora-42-commands-options SUCCESS in 36m 01s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 27m 11s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 6m 28s
system-test-fedora-coreos-next-commands-options NODE_FAILURE Node(set) request 200-0000185883 failed in 0s
system-test-fedora-coreos-next-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000185884 failed in 0s
system-test-fedora-coreos-next-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000185885 failed in 0s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 36m 56s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 34m 00s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 8m 48s
system-test-fedora-coreos-testing-commands-options NODE_FAILURE Node(set) request 200-0000185889 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000185890 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000185891 failed in 0s

@debarshiray

Copy link
Copy Markdown
Member

Hey @Rolv-Apneseth @TristanCacqueray ! Do we have a problem with setting up Fedora CoreOS hosts using the Next and Testing streams on Software Factory? They are consistently encountering NODE_FAILURE. The Stable stream works fine.

@Rolv-Apneseth

Copy link
Copy Markdown
Contributor

Hey @Rolv-Apneseth @TristanCacqueray ! Do we have a problem with setting up Fedora CoreOS hosts using the Next and Testing streams on Software Factory? They are consistently encountering NODE_FAILURE. The Stable stream works fine.

Hey @debarshiray. I imagine there may have been an update breaking these on Software Factory if stable still works fine. Can we get logs for the node failures?

@TristanCacqueray

Copy link
Copy Markdown
Contributor

It looks like SELinux is preventing the image to boot, here are the console log for coreos-next:

Relabeled /sysroot/etc/shadow from <no context> to system_u:object_r:shadow_t:s0                                                                                                                                                                                                         
Relabeled /sysroot/etc/systemd/system/afterburn-sshkeys@core.service from <no context> to system_u:object_r:systemd_unit_file_t:s0                                                                                                                                                       
" Stderr: "setfiles: open(/sysroot/home/zuul-worker) failed: Not a directory                                                                                                                                                                                                             
"[0m                                                                                                                                                                                                                                                                                     
Aug 31 14:12:07 systemd[1]: ignition-files.service: Triggering OnFailure= dependencies.                                                                                                                                                                                                  
                                                                                                                                                                                                                                                                                         
Generating "/run/initramfs/rdsosreport.txt"                                                                                                                                                                                                                                              
                                                                                                                                                                                                                                                                                         
                                                                                                                                                                                                                                                                                         
Entering emergency mode. Exit the shell to continue.                                                                                                                                                                                                                                     
Type "journalctl" to view system logs.                                                                                                                                                                                                                                                   
You might want to save "/run/initramfs/rdsosreport.txt" to a USB stick or /boot                                                                                                                                                                                                          
after mounting them and attach it to a bug report.  

We are still investigating and we'll rebuild the image shortly.

@Rolv-Apneseth

Rolv-Apneseth commented Aug 31, 2026

Copy link
Copy Markdown
Contributor

It's odd these only started failing on 2026-08-22 too, as the latest next and testing builds came out 2026-08-18.

I'll have a look at the butane/ignition and see if there's some kind of mistake there.

@TristanCacqueray

Copy link
Copy Markdown
Contributor

@Rolv-Apneseth images are rebuilt every day. Here is the recipe for coreos: https://gitlab.com/softwarefactory-project/config/-/blob/main/nodepool/dib-ansible/cloud-fedora-coreos-next.yaml?ref_type=heads

The zuul-worker seems to be created with https://gitlab.com/softwarefactory-project/config/-/blob/main/nodepool/providers/fcos-ignition.bu?ref_type=heads , though it's odd that produces a relabeling error.

@TristanCacqueray

Copy link
Copy Markdown
Contributor

Here is the full console log, could it be there is a race condition between ignition-files and the files: ensureUsers ? they seems to be running at the same time.

[    0.000000] Linux version 7.1.8-200.fc44.x86_64 (mockbuild@79d5d8bc7b174d9c91204c7736be3f39) (gcc (GCC) 16.1.1 20260515 (Red Hat 16.1.1-2), GNU ld version 2.46.1-1.fc44) #1 SMP PREEMPT_DYNAMIC Mon Aug 10 03:35:23 UTC 2026
[    0.000000] Command line: BOOT_IMAGE=(hd0,gpt3)/boot/ostree/fedora-coreos-5b206e5a9931097b266bd6176e59fb322b7b3d89459d7adddc943bee3887426d/vmlinuz-7.1.8-200.fc44.x86_64 rw ignition.firstboot mitigations=auto,nosmt ostree=/ostree/boot.1/fedora-coreos/5b206e5a9931097b266bd6176e59fb322b7b3d89459d7adddc943bee3887426d/0 ignition.platform.id=openstack console=ttyS0,115200n8 console=tty0
[    0.000000] BIOS-provided physical RAM map:
[    0.000000] BIOS-e820: [mem 0x0000000000000000-0x000000000009fbff]  System RAM
[    0.000000] BIOS-e820: [mem 0x000000000009fc00-0x000000000009ffff]  device reserved
...
[    4.051751] NetworkManager[739]: <info>  [1788185520.6674] device (ens3): state change: config -> ip-config (reason 'none', managed-type: 'full')
[    4.060453] NetworkManager[739]: <info>  [1788185520.6683] dhcp4 (ens3): activation: beginning transaction (timeout in 90 seconds)
[    4.060479] NetworkManager[739]: <info>  [1788185520.6777] dhcp4 (ens3): state changed new lease, address=38.102.83.113, acd pending
[    4.060504] systemd[1]: Reached target network.target - Network.
[    4.077730] systemd[1]: Starting ignition-fetch.service - Ignition (fetch)...
[    4.090893] systemd[1]: Starting nm-wait-online-initrd.service...
[    4.093416] (ignition)[752]: ignition-fetch.service: Referenced but unset environment variable evaluates to an empty string: IGNITION_ARGS
[    4.112723] ignition[752]: Ignition 2.26.0
[    4.114192] ignition[752]: Stage: fetch
[    4.115573] ignition[752]: reading system config file "/usr/lib/ignition/base.d/00-core.ign"
[    4.118607] ignition[752]: reading system config file "/usr/lib/ignition/base.d/30-afterburn-sshkeys-core.ign"
[    4.122783] ignition[752]: no config dir at "/usr/lib/ignition/base.platform.d/openstack"
[    4.126355] ignition[752]: no config URL provided
[    4.128408] ignition[752]: reading system config file "/usr/lib/ignition/user.ign"
[    4.131719] ignition[752]: no config at "/usr/lib/ignition/user.ign"
[    4.134429] ignition[752]: op(1): [started]  mounting config drive
[    4.137245] ignition[752]: op(1): GET http://169.254.169.254/openstack/latest/user_data: attempt #1
[    4.141284] ignition[752]: op(1): GET http://[fe80::a9fe:a9fe%25ens3]/openstack/latest/user_data: attempt #1
[    4.145109] ignition[752]: op(1): GET error: Get "http://169.254.169.254/openstack/latest/user_data": dial tcp 169.254.169.254:80: connect: network is unreachable
[    4.149757] ignition[752]: op(1): GET error: Get "http://[fe80::a9fe:a9fe%25ens3]/openstack/latest/user_data": dial tcp [fe80::a9fe:a9fe%ens3]:80: connect: cannot assign requested address
[    4.156252] NetworkManager[739]: <info>  [1788185520.8679] dhcp4 (ens3): state changed new lease, address=38.102.83.113
[    4.159779] NetworkManager[739]: <info>  [1788185520.8709] policy: set 'Wired Connection' (ens3) as default for IPv4 routing and DNS
[    4.168349] NetworkManager[739]: <info>  [1788185520.8801] device (ens3): state change: ip-config -> ip-check (reason 'none', managed-type: 'full')
[    4.173776] NetworkManager[739]: <info>  [1788185520.8819] device (ens3): state change: ip-check -> secondaries (reason 'none', managed-type: 'full')
[    4.179027] NetworkManager[739]: <info>  [1788185520.8822] device (ens3): state change: secondaries -> activated (reason 'none', managed-type: 'full')
[    4.184133] NetworkManager[739]: <info>  [1788185520.8826] manager: NetworkManager state is now CONNECTED_SITE
[    4.184236] NetworkManager[739]: <info>  [1788185520.8830] device (ens3): Activation: successful, device activated.
[    4.184404] NetworkManager[739]: <info>  [1788185520.8838] manager: NetworkManager state is now CONNECTED_GLOBAL
[    4.184483] NetworkManager[739]: <info>  [1788185520.8842] manager: startup complete
[    4.198314] ignition[752]: op(1): [finished] mounting config drive
[    4.198349] systemd[1]: Finished nm-wait-online-initrd.service.
[    4.202322] ignition[752]: op(2): [started]  unmounting "/dev/disk/by-label/config-2" at "/tmp/ignition-configdrive1267475551"
[    4.202360] ignition[752]: op(2): [finished] unmounting "/dev/disk/by-label/config-2" at "/tmp/ignition-configdrive1267475551"
[    4.204985] ignition[752]: fetched base config from "system"
[    4.215430] ignition[752]: fetched base config from "system"
[    4.220619] ignition[752]: fetch: fetch complete
[    4.220656] ignition[752]: fetched user config from "openstack"
[    4.220681] ignition[752]: fetch: fetch passed
[    4.220703] ignition[752]: Ignition finished successfully
[    4.227294] systemd[1]: Finished ignition-fetch.service - Ignition (fetch).
[    4.242155] systemd[1]: Starting dracut-initqueue.service - dracut initqueue hook...
[    4.245326] systemd[1]: Starting ignition-ostree-transposefs-detect.service - Ignition OSTree: Detect Partition Transposition...
[    4.290666] systemd[1]: Finished ignition-ostree-transposefs-detect.service - Ignition OSTree: Detect Partition Transposition.
[    4.303743] Key type psk registered
[    4.452040] systemd[1]: Finished dracut-initqueue.service - dracut initqueue hook.
[    4.456036] systemd[1]: Reached target remote-fs-pre.target - Preparation for Remote File Systems.
[    4.459726] systemd[1]: Reached target remote-cryptsetup.target - Remote Encrypted Volumes.
[    4.463241] systemd[1]: Reached target remote-fs.target - Remote File Systems.
[    4.466404] systemd[1]: coreos-livepxe-rootfs.service - Acquire Live PXE rootfs Image skipped, unmet condition check ConditionPathExists=/run/ostree-live
[    4.473344] systemd[1]: coreos-livepxe-persist-osmet.service - Persist Osmet Files (PXE) skipped, unmet condition check ConditionPathExists=/run/ostree-live
[    4.473389] systemd[1]: Starting dracut-pre-mount.service - dracut pre-mount hook...
[    4.508823] systemd[1]: Finished dracut-pre-mount.service - dracut pre-mount hook.
[    4.881940]  vda: vda1 vda2 vda3 vda4
[    4.885096] coreos-gpt-setup[743]: The operation has completed successfully.
[    4.954336] systemd[1]: Finished coreos-gpt-setup.service - Generate New UUID For Boot Disk GPT.
[    4.980922] systemd[1]: ignition-ostree-transposefs-save.service - Ignition OSTree: Save Partitions skipped, unmet condition check ConditionPathIsDirectory=/run/ignition-ostree-transposefs
[    4.989499] systemd[1]: Starting ignition-ostree-uuid-boot.service - Ignition OSTree: Regenerate Filesystem UUID (boot)...
[    5.038446] ignition-ostree-firstboot-uuid[884]: tune2fs 1.47.3 (8-Jul-2025)
[    5.073349] ignition-ostree-firstboot-uuid[882]: Regenerated UUID for /dev/disk/by-label/boot
[    5.081241] systemd[1]: Finished ignition-ostree-uuid-boot.service - Ignition OSTree: Regenerate Filesystem UUID (boot).
[    5.090988] systemd[1]: Starting ignition-kargs.service - Ignition (kargs)...
[    5.107764] (ignition)[889]: ignition-kargs.service: Referenced but unset environment variable evaluates to an empty string: IGNITION_ARGS
[    5.128867] ignition[889]: Ignition 2.26.0
[    5.131207] ignition[889]: Stage: kargs
[    5.133469] ignition[889]: reading system config file "/usr/lib/ignition/base.d/00-core.ign"
[    5.137969] ignition[889]: reading system config file "/usr/lib/ignition/base.d/30-afterburn-sshkeys-core.ign"
[    5.143122] ignition[889]: no config dir at "/usr/lib/ignition/base.platform.d/openstack"
[    5.147449] ignition[889]: kargs: kargs passed
[    5.149850] systemd[1]: Finished ignition-kargs.service - Ignition (kargs).
[    5.149955] ignition[889]: Ignition finished successfully
[    5.157679] systemd[1]: coreos-kargs-reboot.service - CoreOS Kernel Arguments Reboot skipped, unmet condition check ConditionPathExists=/run/coreos-kargs-reboot
[    5.164957] systemd[1]: Starting ignition-disks.service - Ignition (disks)...
[    5.180259] (ignition)[900]: ignition-disks.service: Referenced but unset environment variable evaluates to an empty string: IGNITION_ARGS
[    5.200970] ignition[900]: Ignition 2.26.0
[    5.203289] ignition[900]: Stage: disks
[    5.205412] ignition[900]: reading system config file "/usr/lib/ignition/base.d/00-core.ign"
[    5.209909] ignition[900]: reading system config file "/usr/lib/ignition/base.d/30-afterburn-sshkeys-core.ign"
[    5.215051] systemd[1]: Finished ignition-disks.service - Ignition (disks).
[    5.218682] ignition[900]: no config dir at "/usr/lib/ignition/base.platform.d/openstack"
[    5.218732] ignition[900]: disks: disks passed
[    5.218772] ignition[900]: Ignition finished successfully
[    5.229640] systemd[1]: Reached target initrd-root-device.target - Initrd Root Device.
[    5.236433] systemd[1]: Starting coreos-ignition-unique-boot.service - CoreOS Ignition Ensure Unique Boot Filesystem...
[    5.268329]  vda: vda1 vda2 vda3 vda4
[    5.615817] systemd[1]: Finished coreos-ignition-unique-boot.service - CoreOS Ignition Ensure Unique Boot Filesystem.
[    5.625069] systemd[1]: Starting ignition-ostree-uuid-root.service - Ignition OSTree: Regenerate Filesystem UUID (root)...
[    5.994030] ignition-ostree-firstboot-uuid[931]: Clearing log and setting UUID
[    5.998124] ignition-ostree-firstboot-uuid[931]: writing all SBs
[    6.001501] ignition-ostree-firstboot-uuid[931]: new UUID = 8955dc2e-f9b4-406f-bdf9-5e70f99945a3
[    6.029136] ignition-ostree-firstboot-uuid[926]: Regenerated UUID for /dev/disk/by-label/root
[    6.036687] systemd[1]: Finished ignition-ostree-uuid-root.service - Ignition OSTree: Regenerate Filesystem UUID (root).
[    6.046053] systemd[1]: Starting ignition-ostree-growfs.service - Ignition OSTree: Grow Root Filesystem...
[    6.973927] SGI XFS with ACLs, security attributes, realtime, scrub, repair, quota, no debug enabled
[    6.982197] XFS (vda4): Mounting V5 Filesystem 8955dc2e-f9b4-406f-bdf9-5e70f99945a3
[    7.101769] XFS (vda4): Ending clean mount
[    7.514324] ignition-ostree-growfs[962]: CHANGED: partition=4 start=1050624 old: size=4993024 end=6043647 new: size=82835423 end=83886046
[    8.992536] ignition-ostree-growfs[1019]: meta-data=/dev/vda4              isize=512    agcount=2, agsize=312064 blks
[    8.998231] ignition-ostree-growfs[1019]:          =                       sectsz=512   attr=2, projid32bit=1
[    8.998264] XFS (vda4): Unmounting Filesystem 8955dc2e-f9b4-406f-bdf9-5e70f99945a3
[    8.999336] ignition-ostree-growfs[1019]:          =                       crc=1        finobt=1, sparse=1, rmapbt=1
[    9.013209] ignition-ostree-growfs[1019]:          =                       reflink=1    bigtime=1 inobtcount=1 nrext64=1
[    9.018789] ignition-ostree-growfs[1019]:          =                       exchange=1   metadir=0
[    9.023365] ignition-ostree-growfs[1019]: data     =                       bsize=4096   blocks=624128, imaxpct=25
[    9.023386] ignition-ostree-growfs[1019]:          =                       sunit=0      swidth=0 blks
[    9.023406] ignition-ostree-growfs[1019]: naming   =version 2              bsize=4096   ascii-ci=0, ftype=1, parent=1
[    9.023425] ignition-ostree-growfs[1019]: log      =internal log           bsize=4096   blocks=16384, version=2
[    9.023444] ignition-ostree-growfs[1019]:          =                       sectsz=512   sunit=0 blks, lazy-count=1
[    9.023463] ignition-ostree-growfs[1019]: realtime =none                   extsz=4096   blocks=0, rtextents=0
[    9.023499] ignition-ostree-growfs[1019]:          =                       rgcount=0    rgsize=0 extents
[    9.023518] ignition-ostree-growfs[1019]:          =                       zoned=0      start=0 reserved=0
[    9.023538] ignition-ostree-growfs[1019]: data blocks changed from 624128 to 10354427
[    9.024060] systemd[1]: Finished ignition-ostree-growfs.service - Ignition OSTree: Grow Root Filesystem.
[    9.078457] systemd[1]: Starting ignition-ostree-transposefs-autosave-xfs.service - Ignition OSTree: Autosave XFS Rootfs Partition...
[    9.145765] ignition-ostree-transposefs[1024]: autosave-xfs: /dev/disk/by-label/root agcount=34 is lower than threshold=400
[    9.152172] systemd[1]: Finished ignition-ostree-transposefs-autosave-xfs.service - Ignition OSTree: Autosave XFS Rootfs Partition.
[    9.160379] systemd[1]: ignition-ostree-transposefs-restore.service - Ignition OSTree: Restore Partitions skipped, unmet condition check ConditionPathIsDirectory=/run/ignition-ostree-transposefs
[    9.169002] systemd[1]: Starting coreos-rootflags.service - Determine root FS mount option flags...
[    9.210733] systemd[1]: Finished coreos-rootflags.service - Determine root FS mount option flags.
[    9.218554] systemd[1]: Mounting sysroot.mount - /sysroot...
[    9.247476] XFS (vda4): Mounting V5 Filesystem 8955dc2e-f9b4-406f-bdf9-5e70f99945a3
[    9.292964] XFS (vda4): Ending clean mount
[    9.296966] XFS (vda4): Quotacheck needed: Please wait.
[    9.495698] XFS (vda4): Quotacheck: Done.
[    9.518123] systemd[1]: Mounted sysroot.mount - /sysroot.
[    9.523399] systemd[1]: ignition-remount-sysroot.service - Remount /sysroot read-write for Ignition skipped, unmet condition check ConditionPathIsReadWrite=!/sysroot
[    9.530964] systemd[1]: Starting ostree-prepare-root.service - OSTree Prepare OS/...
[    9.557391] ostree-prepare-root[1077]: Resolved OSTree target to: /sysroot/ostree/deploy/fedora-coreos/deploy/15243a3c5838f3ba255819aa470aa28d6e462ea2fdc97366f00af0917812a240.0
[    9.565413] ostree-prepare-root[1077]: sysroot.readonly configuration value: 1 (fs writable: 1)
[    9.590832] erofs (device erofs): mounted with root inode @ nid 36.
[    9.631186] ostree-prepare-root[1077]: composefs: mounted successfully (verity=false)
[    9.635643] systemd[1]: run-ostree-.private-cfsroot\x2dlower.mount: Deactivated successfully.
[    9.640264] systemd[1]: sysroot.mount: Deactivated successfully.
[    9.643906] ostree-prepare-root[1077]: Resolved OSTree target to: /sysroot.tmp/sysroot/ostree/deploy/fedora-coreos/deploy/15243a3c5838f3ba255819aa470aa28d6e462ea2fdc97366f00af0917812a240.0
[    9.652321] systemd[1]: sysroot.tmp-sysroot.mount: Deactivated successfully.
[    9.652370] systemd[1]: sysroot.tmp-etc.mount: Deactivated successfully.
[    9.652413] systemd[1]: sysroot.tmp.mount: Deactivated successfully.
[    9.652465] systemd[1]: Finished ostree-prepare-root.service - OSTree Prepare OS/.
[    9.669594] systemd[1]: bootc-root-setup.service - bootc setup root skipped, unmet condition check ConditionKernelCommandLine=composefs
[    9.675786] systemd[1]: Reached target initrd-root-fs.target - Initrd Root File System.
[    9.681349] systemd[1]: afterburn-hostname.service - Afterburn Hostname skipped, no trigger condition checks were met.
[    9.686821] systemd[1]: Starting ignition-ostree-check-rootfs-size.service - Ignition OSTree: Check Root Filesystem Size...
[    9.701862] systemd[1]: Starting ignition-ostree-mount-var.service - Mount OSTree /var...
[    9.717804] systemd[1]: Finished ignition-ostree-check-rootfs-size.service - Ignition OSTree: Check Root Filesystem Size.
[    9.718779] ignition-ostree-mount-var[1083]: Mounting /sysroot/sysroot/ostree/deploy/fedora-coreos/var
[    9.730069] systemd[1]: Finished ignition-ostree-mount-var.service - Mount OSTree /var.
[    9.738127] systemd[1]: Starting ignition-mount.service - Ignition (mount)...
[    9.753758] (ignition)[1098]: ignition-mount.service: Referenced but unset environment variable evaluates to an empty string: IGNITION_ARGS
[    9.773734] ignition[1098]: Ignition 2.26.0
[    9.776213] ignition[1098]: Stage: mount
[    9.778446] ignition[1098]: reading system config file "/usr/lib/ignition/base.d/00-core.ign"
[    9.782945] ignition[1098]: reading system config file "/usr/lib/ignition/base.d/30-afterburn-sshkeys-core.ign"
[    9.788124] systemd[1]: Finished ignition-mount.service - Ignition (mount).
[    9.788262] ignition[1098]: no config dir at "/usr/lib/ignition/base.platform.d/openstack"
[    9.788307] ignition[1098]: mount: mount passed
[    9.788348] ignition[1098]: Ignition finished successfully
[    9.802743] systemd[1]: coreos-live-clear-sssd-cache.service - Clear SSSD NSS Cache in Persistent /var skipped, unmet condition check ConditionPathExists=/sysroot/var/lib/sss/mc
[    9.810761] systemd[1]: Starting ignition-ostree-populate-var.service - Populate OSTree /var...
[    9.849081] ignition-ostree-populate-var[1111]: Relabeled /sysroot//var/log from <no context> to system_u:object_r:var_log_t:s0
[    9.938793] ignition-ostree-populate-var[1114]: Relabeled /sysroot//var/home from <no context> to system_u:object_r:home_root_t:s0
[    9.983803] ignition-ostree-populate-var[1117]: Relabeled /sysroot//var/roothome from <no context> to system_u:object_r:admin_home_t:s0
[    9.990004] ignition-ostree-populate-var[1117]: Relabeled /sysroot//var/roothome/buildinfo from <no context> to system_u:object_r:admin_home_t:s0
[    9.996598] ignition-ostree-populate-var[1117]: Relabeled /sysroot//var/roothome/buildinfo/content_manifests from <no context> to system_u:object_r:admin_home_t:s0
[   10.004306] ignition-ostree-populate-var[1117]: Relabeled /sysroot//var/roothome/buildinfo/content_manifests/content-sets.json from <no context> to system_u:object_r:admin_home_t:s0
[   10.012319] ignition-ostree-populate-var[1117]: Relabeled /sysroot//var/roothome/buildinfo/labels.json from <no context> to system_u:object_r:admin_home_t:s0
[   10.019202] ignition-ostree-populate-var[1117]: Relabeled /sysroot//var/roothome/.ssh from <no context> to system_u:object_r:ssh_home_t:s0
[   10.025316] ignition-ostree-populate-var[1117]: Relabeled /sysroot//var/roothome/.bashrc from <no context> to system_u:object_r:admin_home_t:s0
[   10.031671] ignition-ostree-populate-var[1117]: Relabeled /sysroot//var/roothome/.bash_profile from <no context> to system_u:object_r:admin_home_t:s0
[   10.038514] ignition-ostree-populate-var[1117]: Relabeled /sysroot//var/roothome/.bash_logout from <no context> to system_u:object_r:admin_home_t:s0
[   10.045175] ignition-ostree-populate-var[1120]: Relabeled /sysroot//var/opt from <no context> to system_u:object_r:usr_t:s0
[   10.050983] ignition-ostree-populate-var[1123]: Relabeled /sysroot//var/srv from <no context> to system_u:object_r:var_t:s0
[   10.079086] ignition-ostree-populate-var[1126]: Relabeled /sysroot//var/usrlocal from <no context> to system_u:object_r:usr_t:s0
[   10.116105] ignition-ostree-populate-var[1129]: Relabeled /sysroot//var/mnt from <no context> to system_u:object_r:mnt_t:s0
[   10.146172] systemd[1]: Finished ignition-ostree-populate-var.service - Populate OSTree /var.
[   10.153550] systemd[1]: Starting ignition-ostree-mount-state-overlays.service - Ignition OSTree Mount State Overlays...
[   10.176209] systemd[1]: Finished ignition-ostree-mount-state-overlays.service - Ignition OSTree Mount State Overlays.
[   10.185307] systemd[1]: Starting ignition-files.service - Ignition (files)...
[   10.198539] (ignition)[1137]: ignition-files.service: Referenced but unset environment variable evaluates to an empty string: IGNITION_ARGS
[   10.220449] ignition[1137]: Ignition 2.26.0
[   10.222780] ignition[1137]: Stage: files
[   10.225028] ignition[1137]: reading system config file "/usr/lib/ignition/base.d/00-core.ign"
[   10.229629] ignition[1137]: reading system config file "/usr/lib/ignition/base.d/30-afterburn-sshkeys-core.ign"
[   10.235381] ignition[1137]: no config dir at "/usr/lib/ignition/base.platform.d/openstack"
[   10.249524] ignition[1137]: files: ensureUsers: op(1): [started]  creating or modifying user "core"
[   10.303809] ignition[1137]: files: ensureUsers: op(1): [finished] creating or modifying user "core"
[   10.306938] ignition[1137]: files: ensureUsers: op(2): [started]  creating or modifying user "zuul-worker"
[   10.340883] ignition[1137]: files: ensureUsers: op(2): [finished] creating or modifying user "zuul-worker"
[   10.344128] ignition[1137]: files: createFilesystemsFiles: createFiles: op(3): [started]  writing file "/sysroot/etc/sudoers.d/zuul-worker"
[   10.348563] ignition[1137]: files: createFilesystemsFiles: createFiles: op(3): [finished] writing file "/sysroot/etc/sudoers.d/zuul-worker"
[   10.352577] ignition[1137]: files: op(4): [started]  processing unit "afterburn-sshkeys@core.service"
[   10.355817] ignition[1137]: files: op(4): [finished] processing unit "afterburn-sshkeys@core.service"
[   10.358893] ignition[1137]: files: op(5): [started]  masking unit "afterburn-sshkeys@core.service"
[   10.361861] ignition[1137]: files: op(5): [finished] masking unit "afterburn-sshkeys@core.service"
[   10.365076] ignition[1137]: files: op(6): [started]  processing unit "afterburn-sshkeys@zuul-worker.service"
[   10.368492] ignition[1137]: files: op(6): [finished] processing unit "afterburn-sshkeys@zuul-worker.service"
[   10.371865] ignition[1137]: files: op(7): [started]  processing unit "zincati.service"
[   10.374645] ignition[1137]: files: op(7): [finished] processing unit "zincati.service"
[   10.377368] ignition[1137]: files: op(8): [started]  masking unit "zincati.service"
[   10.380051] ignition[1137]: files: op(8): [finished] masking unit "zincati.service"
[   10.382637] ignition[1137]: files: op(9): [started]  processing unit "zuul-firstboot.service"
[   10.385502] ignition[1137]: files: op(9): op(a): [started]  writing unit "zuul-firstboot.service" at "/sysroot/etc/systemd/system/zuul-firstboot.service"
[   10.389798] ignition[1137]: files: op(9): op(a): [finished] writing unit "zuul-firstboot.service" at "/sysroot/etc/systemd/system/zuul-firstboot.service"
[   10.394065] ignition[1137]: files failed
[   10.395454] ignition[1137]: Full config:
[   10.395470] ignition[1137]: {
[   10.398197] ignition[1137]:   "ignition": {
[   10.398212] ignition[1137]:     "config": {
[   10.401332] ignition[1137]:       "replace": {
[   10.401347] ignition[1137]:         "verification": {}
[   10.401361] ignition[1137]:       }
[   10.401375] ignition[1137]:     },
[   10.401389] ignition[1137]:     "proxy": {},
[   10.401402] ignition[1137]:     "security": {
[   10.401417] ignition[1137]:       "tls": {}
[   10.401431] ignition[1137]:     },
[   10.401444] ignition[1137]:     "timeouts": {},
[   10.401463] ignition[1137]:     "version": "3.7.0-experimental"
[   10.401477] ignition[1137]:   },
[   10.401494] ignition[1137]:   "kernelArguments": {},
[   10.401508] ignition[1137]:   "passwd": {
[   10.401521] ignition[1137]:     "users": [
[   10.401535] ignition[1137]:       {
[   10.401551] ignition[1137]:         "gecos": "CoreOS Admin",
[   10.401565] ignition[1137]:         "groups": [
[   10.401579] ignition[1137]:           "adm",
[   10.401593] ignition[1137]:           "sudo",
[   10.401607] ignition[1137]:           "systemd-journal",
[   10.401621] ignition[1137]:           "wheel"
[   10.401634] ignition[1137]:         ],
[   10.401648] ignition[1137]:         "name": "core"
[   10.401662] ignition[1137]:       },
[   10.401675] ignition[1137]:       {
[   10.401689] ignition[1137]:         "groups": [
[   10.401702] ignition[1137]:           "wheel"
[   10.401716] ignition[1137]:         ],
[   10.401731] ignition[1137]:         "homeDir": "/home/zuul-worker",
[   10.401746] ignition[1137]:         "name": "zuul-worker"
[   10.401759] ignition[1137]:       }
[   10.401773] ignition[1137]:     ]
[   10.442479] ignition[1137]:   },
[   10.442490] ignition[1137]:   "storage": {
[   10.442501] ignition[1137]:     "files": [
[   10.442511] ignition[1137]:       {
[   10.446411] ignition[1137]:         "group": {},
[   10.446422] ignition[1137]:         "path": "/etc/sudoers.d/zuul-worker",
[   10.446432] ignition[1137]:         "user": {},
[   10.446442] ignition[1137]:         "contents": {
[   10.446456] ignition[1137]:           "compression": "",
[   10.446467] ignition[1137]:           "source": "data:,zuul-worker%20ALL%3D(ALL)%20NOPASSWD%3AALL%0A",
[   10.446477] ignition[1137]:           "verification": {}
[   10.446487] ignition[1137]:         },
[   10.446497] ignition[1137]:         "mode": 288
[   10.446507] ignition[1137]:       }
[   10.446516] ignition[1137]:     ]
[   10.446526] ignition[1137]:   },
[   10.446535] ignition[1137]:   "systemd": {
[   10.446546] ignition[1137]:     "units": [
[   10.446556] ignition[1137]:       {
[   10.446565] ignition[1137]:         "enabled": true,
[   10.446576] ignition[1137]:         "mask": true,
[   10.446588] ignition[1137]:         "name": "afterburn-sshkeys@core.service"
[   10.446598] ignition[1137]:       },
[   10.446607] ignition[1137]:       {
[   10.446617] ignition[1137]:         "enabled": true,
[   10.446626] ignition[1137]:         "name": "afterburn-sshkeys@zuul-worker.service"
[   10.446637] ignition[1137]:       },
[   10.446648] ignition[1137]:       {
[   10.446657] ignition[1137]:         "mask": true,
[   10.446667] ignition[1137]:         "name": "zincati.service"
[   10.446677] ignition[1137]:       },
[   10.446687] ignition[1137]:       {
[   10.446699] ignition[1137]:         "contents": "[Unit]\nDescription=Install Zuul dependencies via rpm-ostree\nConditionFirstBoot=true\nBefore=sshd.service\nAfter=network-online.target\nWants=network-online.target\n\n[Service]\nType=oneshot\nRemainAfterExit=true\nExecStart=/usr/bin/afterburn --cmdline --hostname=/etc/hostname\nExecStart=/usr/bin/rpm-ostree install --apply-live -y python3 python3-libselinux rsync git\n\n[Install]\nWantedBy=multi-user.target\n",
[   10.488350] ignition[1137]:         "enabled": true,
[   10.488361] ignition[1137]:         "name": "zuul-firstboot.service"
[   10.488372] ignition[1137]:       },
[   10.488381] ignition[1137]:       {
[   10.493197] ignition[1137]:         "dropins": [
[   10.493208] ignition[1137]:           {
[   10.493219] ignition[1137]:             "contents": "[Unit]\nAfter=zuul-firstboot.service afterburn-sshkeys@zuul-worker.service\nRequires=zuul-firstboot.service afterburn-sshkeys@zuul-worker.service\n",
[   10.493249] ignition[1137]: files: op(9): [finished] processing unit "zuul-firstboot.service"
[   10.501867] systemd[1]: ignition-files.service: Main process exited, code=exited, status=1/FAILURE
[   10.503998] ignition[1137]:             "name": "10-wait-for-zuul-firstboot.conf"
[   10.505942] ignition[1137]:           }
[   10.505953] ignition[1137]:         ],
[   10.505962] ignition[1137]:         "name": "sshd.service"
[   10.505972] ignition[1137]:       }
[   10.505982] ignition[1137]:     ]
[   10.505991] ignition[1137]:   }
[   10.506001] ignition[1137]: }
[   10.506023] ignition[1137]: files: op(b): [started]  processing unit "sshd.service"
[   10.506049] systemd[1]: ignition-files.service: Failed with result 'exit-code'.
[   10.506100] ignition[1137]: files: op(b): op(c): [started]  writing systemd drop-in "10-wait-for-zuul-firstboot.conf" at "/sysroot/etc/systemd/system/sshd.service.d/10-wait-for-zuul-firstboot.conf"
[   10.506152] systemd[1]: Failed to start ignition-files.service - Ignition (files).
[   10.506181] ignition[1137]: files: op(b): op(c): [finished] writing systemd drop-in "10-wait-for-zuul-firstboot.conf" at "/sysroot/etc/systemd/system/sshd.service.d/10-wait-for-zuul-firstboot.conf"
[   10.506204] ignition[1137]: files: op(b): [finished] processing unit "sshd.service"
[   10.506224] ignition[1137]: files: op(d): [started]  setting preset to enabled for "afterburn-sshkeys@.service core zuul-worker"
[   10.506242] ignition[1137]: files: op(d): [finished] setting preset to enabled for "afterburn-sshkeys@.service core zuul-worker"
[   10.506260] ignition[1137]: files: op(e): [started]  setting preset to enabled for "zuul-firstboot.service"
[   10.506279] ignition[1137]: files: op(e): [finished] setting preset to enabled for "zuul-firstboot.service"
[   10.506297] ignition[1137]: files: createResultFile: createFiles: op(f): [started]  writing file "/sysroot/etc/.ignition-result.json"
[   10.506316] ignition[1137]: files: createResultFile: createFiles: op(f): [finished] writing file "/sysroot/etc/.ignition-result.json"
[   10.506334] ignition[1137]: files: op(10): [started]  relabeling 22 patterns
[   10.586947] systemd[1]: Dependency failed for ignition-complete.target - Ignition Complete.
[   10.590376] systemd[1]: Dependency failed for initrd.target - Initrd Default Target.
[   10.593548] systemd[1]: initrd.target: Job initrd.target/start failed with result 'dependency'.
[   10.598481] systemd[1]: initrd.target: Triggering OnFailure= dependencies.
[   10.602467] systemd[1]: ignition-complete.target: Job ignition-complete.target/start failed with result 'dependency'.
[   10.608424] systemd[1]: ignition-files.service: Triggering OnFailure= dependencies.
[   10.608484] systemd[1]: Stopped target initrd-root-device.target - Initrd Root Device.
[   10.614331] systemd[1]: Stopped target initrd-usr-fs.target - Initrd /usr File System.
[   10.617661] systemd[1]: Stopped target remote-cryptsetup.target - Remote Encrypted Volumes.
[   10.621159] systemd[1]: Stopped target remote-fs.target - Remote File Systems.
[   10.624191] systemd[1]: Stopped target remote-fs-pre.target - Preparation for Remote File Systems.
[   10.630081] systemd[1]: Stopped target timers.target - Timer Units.
[   10.632817] systemd[1]: coreos-rootflags.service: Deactivated successfully.
[   10.636742] systemd[1]: Stopped coreos-rootflags.service - Determine root FS mount option flags.
[   10.640405] systemd[1]: coreos-touch-run-agetty.service: Deactivated successfully.
[   10.645089] systemd[1]: Stopped coreos-touch-run-agetty.service - CoreOS: Touch /run/agetty.reload.
[   10.648872] systemd[1]: ignition-ostree-check-rootfs-size.service: Deactivated successfully.
[   10.653684] systemd[1]: Stopped ignition-ostree-check-rootfs-size.service - Ignition OSTree: Check Root Filesystem Size.
[   10.659290] systemd[1]: Stopping ignition-ostree-mount-state-overlays.service - Ignition OSTree Mount State Overlays...
[   10.663446] systemd[1]: ignition-ostree-transposefs-autosave-xfs.service: Deactivated successfully.
[   10.668575] systemd[1]: Stopped ignition-ostree-transposefs-autosave-xfs.service - Ignition OSTree: Autosave XFS Rootfs Partition.
[   10.682318] systemd[1]: ignition-ostree-mount-state-overlays.service: Deactivated successfully.
[   10.687404] systemd[1]: Stopped ignition-ostree-mount-state-overlays.service - Ignition OSTree Mount State Overlays.
[   10.691832] systemd[1]: ignition-ostree-populate-var.service: Deactivated successfully.
[   10.696502] systemd[1]: Stopped ignition-ostree-populate-var.service - Populate OSTree /var.
[   10.701094] systemd[1]: Stopping ignition-mount.service - Ignition (mount)...
[   10.711991] (ignition)[1163]: ignition-mount.service: Referenced but unset environment variable evaluates to an empty string: IGNITION_ARGS
[   10.730003] ignition[1163]: Ignition 2.26.0
[   10.732586] ignition[1163]: Stage: umount
[   10.735015] ignition[1163]: reading system config file "/usr/lib/ignition/base.d/00-core.ign"
[   10.747500] systemd[1]: ignition-mount.service: Deactivated successfully.
[   10.747631] systemd[1]: Stopped ignition-mount.service - Ignition (mount).
[   10.747679] ignition[1163]: reading system config file "/usr/lib/ignition/base.d/30-afterburn-sshkeys-core.ign"
[   10.747770] ignition[1163]: no config dir at "/usr/lib/ignition/base.platform.d/openstack"
[   10.765185] ignition[1163]: umount: umount passed
[   10.765235] ignition[1163]: Ignition finished successfully
[   10.770993] systemd[1]: Stopping ignition-ostree-mount-var.service - Mount OSTree /var...
[   10.789927] ignition-ostree-mount-var[1173]: Unmounting /sysroot/var
[   10.793623] systemd[1]: sysroot-var.mount: Deactivated successfully.
[   10.797049] systemd[1]: ignition-ostree-mount-var.service: Deactivated successfully.
[   10.797234] systemd[1]: Stopped ignition-ostree-mount-var.service - Mount OSTree /var.
[   10.803411] systemd[1]: Stopped target initrd-root-fs.target - Initrd Root File System.
[   10.806799] systemd[1]: dracut-pre-mount.service: Deactivated successfully.
[   10.810562] systemd[1]: Stopped dracut-pre-mount.service - dracut pre-mount hook.
[   10.813706] systemd[1]: dracut-initqueue.service: Deactivated successfully.
[   10.817363] systemd[1]: Stopped dracut-initqueue.service - dracut initqueue hook.
[   10.820695] systemd[1]: ignition-ostree-growfs.service: Deactivated successfully.
[   10.824771] systemd[1]: Stopped ignition-ostree-growfs.service - Ignition OSTree: Grow Root Filesystem.
[   10.828625] systemd[1]: ignition-ostree-uuid-root.service: Deactivated successfully.
[   10.832830] systemd[1]: Stopped ignition-ostree-uuid-root.service - Ignition OSTree: Regenerate Filesystem UUID (root).
[   10.837050] systemd[1]: coreos-ignition-unique-boot.service: Deactivated successfully.
[   10.841339] systemd[1]: Stopped coreos-ignition-unique-boot.service - CoreOS Ignition Ensure Unique Boot Filesystem.
[   10.845476] systemd[1]: ignition-disks.service: Deactivated successfully.
[   10.849066] systemd[1]: Stopped ignition-disks.service - Ignition (disks).
[   10.852196] systemd[1]: ignition-kargs.service: Deactivated successfully.
[   10.855818] systemd[1]: Stopped ignition-kargs.service - Ignition (kargs).
[   10.861914] systemd[1]: Stopping ignition-ostree-transposefs-detect.service - Ignition OSTree: Detect Partition Transposition...
[   10.866214] systemd[1]: ignition-ostree-uuid-boot.service: Deactivated successfully.
[   10.870552] systemd[1]: Stopped ignition-ostree-uuid-boot.service - Ignition OSTree: Regenerate Filesystem UUID (boot).
[   10.874818] systemd[1]: coreos-gpt-setup.service: Deactivated successfully.
[   10.878530] systemd[1]: Stopped coreos-gpt-setup.service - Generate New UUID For Boot Disk GPT.
[   10.882097] systemd[1]: coreos-unique-boot.service: Deactivated successfully.
[   10.886112] systemd[1]: Stopped coreos-unique-boot.service - CoreOS Ensure Unique Boot Filesystem Label.
[   10.889838] systemd[1]: nm-wait-online-initrd.service: Deactivated successfully.
[   10.894298] systemd[1]: Stopped nm-wait-online-initrd.service.
[   10.896809] systemd[1]: ostree-prepare-root.service: Deactivated successfully.
[   10.901166] systemd[1]: Stopped ostree-prepare-root.service - OSTree Prepare OS/.
[   10.905980] systemd[1]: ignition-ostree-transposefs-detect.service: Deactivated successfully.
[   10.910970] systemd[1]: Stopped ignition-ostree-transposefs-detect.service - Ignition OSTree: Detect Partition Transposition.
[   10.915949] systemd[1]: ignition-fetch.service: Deactivated successfully.
[   10.919911] systemd[1]: Stopped ignition-fetch.service - Ignition (fetch).
[   10.922954] systemd[1]: Stopped target network.target - Network.
[   10.925602] systemd[1]: Stopping nm-initrd.service...
[   10.926387] NetworkManager[739]: <info>  [1788185527.6375] caught SIGTERM, shutting down normally.
[   10.933388] NetworkManager[739]: <info>  [1788185527.6451] dhcp4 (ens3): canceled DHCP transaction
[   10.940928] NetworkManager[739]: <info>  [1788185527.6452] dhcp4 (ens3): activation: beginning transaction (timeout in 90 seconds)
[   10.947472] NetworkManager[739]: <info>  [1788185527.6452] dhcp4 (ens3): state changed no lease
[   10.952372] NetworkManager[739]: <info>  [1788185527.6455] manager: NetworkManager state is now CONNECTED_SITE
[   10.957630] NetworkManager[739]: <info>  [1788185527.6462] exiting (success)
[   10.961272] systemd[1]: nm-initrd.service: Deactivated successfully.
[   10.961331] systemd[1]: Stopped nm-initrd.service.
[   10.965717] systemd[1]: coreos-copy-firstboot-network.service: Deactivated successfully.
[   10.970025] systemd[1]: Stopped coreos-copy-firstboot-network.service - Copy CoreOS Firstboot Networking Config.
[   10.973992] systemd[1]: coreos-enable-network.service: Deactivated successfully.
[   10.978017] systemd[1]: Stopped coreos-enable-network.service - CoreOS Enable Network.
[   10.981455] systemd[1]: ignition-fetch-offline.service: Deactivated successfully.
[   10.985531] systemd[1]: Stopped ignition-fetch-offline.service - Ignition (fetch-offline).
[   10.989051] systemd[1]: Stopped target basic.target - Basic System.
[   10.991742] systemd[1]: Stopped target paths.target - Path Units.
[   10.994352] systemd[1]: Stopped target slices.target - Slice Units.
[   10.996959] systemd[1]: Stopped target sockets.target - Socket Units.
[   10.999691] systemd[1]: Stopped target sysinit.target - System Initialization.
[   11.002786] systemd[1]: Stopped target cryptsetup.target - Local Encrypted Volumes.
[   11.005950] systemd[1]: systemd-ask-password-console.path: Deactivated successfully.
[   11.010130] systemd[1]: Stopped systemd-ask-password-console.path - Dispatch Password Requests to Console Directory Watch.
[   11.014477] systemd[1]: Stopped target cryptsetup-pre.target - Local Encrypted Volumes (Pre).
[   11.017903] systemd[1]: clevis-luks-askpass.path: Deactivated successfully.
[   11.021535] systemd[1]: Stopped clevis-luks-askpass.path - Forward Password Requests to Clevis Directory Watch.
[   11.025573] systemd[1]: Stopped target local-fs.target - Local File Systems.
[   11.028480] systemd[1]: Stopped target local-fs-pre.target - Preparation for Local File Systems.
[   11.032054] systemd[1]: Stopped target swap.target - Swaps.
[   11.034382] systemd[1]: iscsid.socket: Deactivated successfully.
[   11.037523] systemd[1]: Closed iscsid.socket - Open-iSCSI iscsid Socket.
[   11.040537] systemd[1]: iscsiuio.socket: Deactivated successfully.
[   11.043816] systemd[1]: Closed iscsiuio.socket - Open-iSCSI iscsiuio Socket.
[   11.046753] systemd[1]: coreos-ignition-setup-user.service: Deactivated successfully.
[   11.050987] systemd[1]: Stopped coreos-ignition-setup-user.service - CoreOS Ignition User Config Setup.
[   11.055976] dbus-broker[748]: Dispatched 1658 messages @ 1(��2)��s / message.
[   11.059740] systemd[1]: Stopping dbus-broker.service - D-Bus System Message Bus...
[   11.063831] systemd[1]: Started emergency.service - Emergency Shell.
[   11.066518] systemd[1]: Reached target emergency.target - Emergency Mode.
[   11.070376] systemd[1]: Starting ignition-virtio-dump-journal.service - Dump Journal to Virtio Port...
[   11.074049] systemd[1]: systemd-sysctl.service: Deactivated successfully.
[   11.074105] systemd[1]: Stopped systemd-sysctl.service - Apply Kernel Variables.
[   11.079288] systemd[1]: systemd-modules-load.service: Deactivated successfully.
[   11.079344] systemd[1]: Stopped systemd-modules-load.service - Load Kernel Modules.
[   11.089291] systemd[1]: systemd-udev-trigger.service: Deactivated successfully.
[   11.093294] systemd[1]: Stopped systemd-udev-trigger.service - Coldplug All udev Devices.
[   11.093631] ignition-virtio-dump-journal[1181]: Didn't find virtio port /dev/virtio-ports/com.coreos.ignition.journal
[   11.103185] systemd[1]: systemd-udevd-control.socket: Deactivated successfully.
[   11.103244] systemd[1]: Closed systemd-udevd-control.socket - udev Control Socket.
[   11.108735] systemd[1]: dracut-pre-trigger.service: Deactivated successfully.
[   11.112443] systemd[1]: Stopped dracut-pre-trigger.service - dracut pre-trigger hook.
[   11.115620] systemd[1]: dracut-pre-udev.service: Deactivated successfully.
[   11.115681] systemd[1]: Stopped dracut-pre-udev.service - dracut pre-udev hook.
[   11.120786] systemd[1]: dracut-cmdline.service: Deactivated successfully.
[   11.120839] systemd[1]: Stopped dracut-cmdline.service - dracut cmdline hook.
[   11.125758] systemd[1]: afterburn-network-kargs.service: Deactivated successfully.
[   11.125812] systemd[1]: Stopped afterburn-network-kargs.service - Afterburn Initrd Setup Network Kernel Arguments.
[   11.132130] systemd[1]: dracut-cmdline-ask.service: Deactivated successfully.
[   11.132207] systemd[1]: Stopped dracut-cmdline-ask.service - dracut ask for additional cmdline parameters.
[   11.138191] systemd[1]: systemd-tmpfiles-setup-dev.service: Deactivated successfully.
[   11.138254] systemd[1]: Stopped systemd-tmpfiles-setup-dev.service - Create Static Device Nodes in /dev.
[   11.144407] systemd[1]: systemd-tmpfiles-setup-dev-early.service: Deactivated successfully.
[   11.144459] systemd[1]: Stopped systemd-tmpfiles-setup-dev-early.service - Create Static Device Nodes in /dev gracefully.
[   11.150472] systemd[1]: kmod-static-nodes.service: Deactivated successfully.
[   11.150532] systemd[1]: Stopped kmod-static-nodes.service - Create List of Static Device Nodes.
[   11.158779] systemd[1]: dbus-broker.service: Deactivated successfully.
[   11.162227] systemd[1]: Stopped dbus-broker.service - D-Bus System Message Bus.
[   11.165654] systemd[1]: Finished ignition-virtio-dump-journal.service - Dump Journal to Virtio Port.
[   11.169097] systemd[1]: dbus.socket: Deactivated successfully.
[   11.172269] systemd[1]: Closed dbus.socket - D-Bus System Message Bus Socket.
[   11.175126] systemd[1]: systemd-tmpfiles-setup.service: Deactivated successfully.
[   11.179212] systemd[1]: Stopped systemd-tmpfiles-setup.service - Create System Files and Directories.
[   11.182955] systemd[1]: Startup finished in 1.210s (kernel) + 0 (initrd) + 9.790s (userspace) = 11.001s.
------
Ignition has failed. Please ensure your config is valid. Note that only
Ignition spec v3.0.0+ configs are accepted.

A CLI validation tool to check this called ignition-validate can be
downloaded from GitHub:
    https://github.com/coreos/ignition/releases
------

Displaying logs from failed units: ignition-files.service
Aug 31 14:12:07 ignition[1137]: files: op(d): [finished] setting preset to enabled for "afterburn-sshkeys@.service core zuul-worker"
Aug 31 14:12:07 ignition[1137]: files: op(e): [started]  setting preset to enabled for "zuul-firstboot.service"
Aug 31 14:12:07 ignition[1137]: files: op(e): [finished] setting preset to enabled for "zuul-firstboot.service"
Aug 31 14:12:07 ignition[1137]: files: createResultFile: createFiles: op(f): [started]  writing file "/sysroot/etc/.ignition-result.json"
Aug 31 14:12:07 ignition[1137]: files: createResultFile: createFiles: op(f): [finished] writing file "/sysroot/etc/.ignition-result.json"
Aug 31 14:12:07 ignition[1137]: files: op(10): [started]  relabeling 22 patterns
Aug 31 14:12:07 ignition[1137]: [0;38:5:245m[0;1;31m[0;38:5:245mfiles: op(10): executing: "setfiles" "-vF0" "-r" "/sysroot" "/sysroot/etc/selinux/targeted/contexts/files/file_contexts" "-f" "-"[0m
Aug 31 14:12:07 ignition[1137]: [0;1;31m[0;1;39m[0;1;31mfiles: op(10): [failed]   relabeling 22 patterns: exit status 255: Cmd: "setfiles" "-vF0" "-r" "/sysroot" "/sysroot/etc/selinux/targeted/contexts/files/file_contexts" "-f" "-" Stdout: "Relabeled /sysroot/etc/systemd/system/zincati.service from <no context> to system_u:object_r:systemd_unit_file_t:s0
Relabeled /sysroot/etc/.ignition-result.json from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/subuid- from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/var/home/core from <no context> to unconfined_u:object_r:user_home_dir_t:s0
Relabeled /sysroot/var/home/core/.bash_logout from <no context> to unconfined_u:object_r:user_home_t:s0
Relabeled /sysroot/var/home/core/.bash_profile from <no context> to unconfined_u:object_r:user_home_t:s0
Relabeled /sysroot/var/home/core/.bashrc from <no context> to unconfined_u:object_r:user_home_t:s0
Relabeled /sysroot/etc/systemd/system/zuul-firstboot.service from <no context> to system_u:object_r:systemd_unit_file_t:s0
Relabeled /sysroot/etc/group- from <no context> to system_u:object_r:passwd_file_t:s0
Relabeled /sysroot/etc/gshadow from <no context> to system_u:object_r:shadow_t:s0
Relabeled /sysroot/etc/gshadow- from <no context> to system_u:object_r:shadow_t:s0
Relabeled /sysroot/etc/subuid from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/.pwd.lock from <no context> to system_u:object_r:passwd_file_t:s0
Relabeled /sysroot/etc/systemd/system/sshd.service.d from <no context> to system_u:object_r:systemd_unit_file_t:s0
Relabeled /sysroot/etc/systemd/system/sshd.service.d/10-wait-for-zuul-firstboot.conf from <no context> to system_u:object_r:sshd_unit_file_t:s0
Relabeled /sysroot/etc/passwd- from <no context> to system_u:object_r:passwd_file_t:s0
Relabeled /sysroot/etc/group from <no context> to system_u:object_r:passwd_file_t:s0
Relabeled /sysroot/etc/shadow- from <no context> to system_u:object_r:shadow_t:s0
Relabeled /sysroot/etc/subgid from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/subgid- from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/sudoers.d/zuul-worker from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/systemd/system-preset from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/systemd/system-preset/20-ignition.preset from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/passwd from <no context> to system_u:object_r:passwd_file_t:s0
Relabeled /sysroot/etc/shadow from <no context> to system_u:object_r:shadow_t:s0
Relabeled /sysroot/etc/systemd/system/afterburn-sshkeys@core.service from <no context> to system_u:object_r:systemd_unit_file_t:s0
" Stderr: "setfiles: open(/sysroot/home/zuul-worker) failed: Not a directory
"[0m
Aug 31 14:12:07 ignition[1137]: [0;1;31m[0;1;39m[0;1;31mIgnition failed: failed to handle relabeling: exit status 255: Cmd: "setfiles" "-vF0" "-r" "/sysroot" "/sysroot/etc/selinux/targeted/contexts/files/file_contexts" "-f" "-" Stdout: "Relabeled /sysroot/etc/systemd/system/zincati.service from <no context> to system_u:object_r:systemd_unit_file_t:s0
Relabeled /sysroot/etc/.ignition-result.json from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/subuid- from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/var/home/core from <no context> to unconfined_u:object_r:user_home_dir_t:s0
Relabeled /sysroot/var/home/core/.bash_logout from <no context> to unconfined_u:object_r:user_home_t:s0
Relabeled /sysroot/var/home/core/.bash_profile from <no context> to unconfined_u:object_r:user_home_t:s0
Relabeled /sysroot/var/home/core/.bashrc from <no context> to unconfined_u:object_r:user_home_t:s0
Relabeled /sysroot/etc/systemd/system/zuul-firstboot.service from <no context> to system_u:object_r:systemd_unit_file_t:s0
Relabeled /sysroot/etc/group- from <no context> to system_u:object_r:passwd_file_t:s0
Relabeled /sysroot/etc/gshadow from <no context> to system_u:object_r:shadow_t:s0
Relabeled /sysroot/etc/gshadow- from <no context> to system_u:object_r:shadow_t:s0
Relabeled /sysroot/etc/subuid from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/.pwd.lock from <no context> to system_u:object_r:passwd_file_t:s0
Relabeled /sysroot/etc/systemd/system/sshd.service.d from <no context> to system_u:object_r:systemd_unit_file_t:s0
Relabeled /sysroot/etc/systemd/system/sshd.service.d/10-wait-for-zuul-firstboot.conf from <no context> to system_u:object_r:sshd_unit_file_t:s0
Relabeled /sysroot/etc/passwd- from <no context> to system_u:object_r:passwd_file_t:s0
Relabeled /sysroot/etc/group from <no context> to system_u:object_r:passwd_file_t:s0
Relabeled /sysroot/etc/shadow- from <no context> to system_u:object_r:shadow_t:s0
Relabeled /sysroot/etc/subgid from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/subgid- from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/sudoers.d/zuul-worker from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/systemd/system-preset from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/systemd/system-preset/20-ignition.preset from <no context> to system_u:object_r:etc_t:s0
Relabeled /sysroot/etc/passwd from <no context> to system_u:object_r:passwd_file_t:s0
Relabeled /sysroot/etc/shadow from <no context> to system_u:object_r:shadow_t:s0
Relabeled /sysroot/etc/systemd/system/afterburn-sshkeys@core.service from <no context> to system_u:object_r:systemd_unit_file_t:s0
" Stderr: "setfiles: open(/sysroot/home/zuul-worker) failed: Not a directory
"[0m
Aug 31 14:12:07 systemd[1]: ignition-files.service: Triggering OnFailure= dependencies.

Generating "/run/initramfs/rdsosreport.txt"


Entering emergency mode. Exit the shell to continue.
Type "journalctl" to view system logs.
You might want to save "/run/initramfs/rdsosreport.txt" to a USB stick or /boot
after mounting them and attach it to a bug report.

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/28e97fa4bec142698e5214f044998d6a

✔️ unit-test SUCCESS in 2m 21s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 51s
✔️ unit-test-restricted SUCCESS in 2m 23s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 48m 38s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 33m 40s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 8m 02s
✔️ system-test-fedora-44-commands-options SUCCESS in 47m 03s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 37m 53s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 7m 44s
✔️ system-test-fedora-43-commands-options SUCCESS in 54m 02s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 32m 45s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 6m 55s
✔️ system-test-fedora-42-commands-options SUCCESS in 39m 16s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 23m 05s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 6m 10s
✔️ system-test-fedora-coreos-next-commands-options SUCCESS in 46m 48s
✔️ system-test-fedora-coreos-next-runtime-environment-arch-fedora SUCCESS in 42m 20s
✔️ system-test-fedora-coreos-next-runtime-environment-ubuntu SUCCESS in 11m 45s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 56m 47s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 42m 34s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 12m 52s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 52m 22s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 42m 32s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 12m 13s

debarshiray pushed a commit to yselkowitz/toolbox that referenced this pull request Sep 3, 2026
This allows using the 'distro' option to create and enter Fedora ELN
containers.  Due to ELN's rolling-release model, the 'release' option
isn't required.  If 'release' is used, the accepted value is 'latest'.

Note that the VERSION_ID field in os-release(5) on Fedora ELN contains a
number representing the next major version of Red Hat Enterprise Linux,
even though it follows a rolling-release model.  Right now it is:
  VERSION_ID=11

containers#1814

Signed-off-by: Yaakov Selkowitz <yselkowi@redhat.com>
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/bb204753a5c64dce80d48be5729694c8

✔️ unit-test SUCCESS in 2m 21s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 41s
✔️ unit-test-restricted SUCCESS in 2m 15s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 46m 44s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 33m 35s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 8m 39s
✔️ system-test-fedora-44-commands-options SUCCESS in 53m 49s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 31m 49s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 9m 39s
✔️ system-test-fedora-43-commands-options SUCCESS in 58m 32s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 41m 54s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 8m 38s
✔️ system-test-fedora-42-commands-options SUCCESS in 53m 27s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 26m 59s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 8m 06s
✔️ system-test-fedora-coreos-next-commands-options SUCCESS in 58m 05s
✔️ system-test-fedora-coreos-next-runtime-environment-arch-fedora SUCCESS in 47m 43s
✔️ system-test-fedora-coreos-next-runtime-environment-ubuntu SUCCESS in 16m 14s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 57m 49s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 47m 50s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 16m 05s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 58m 50s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 43m 40s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 10m 55s

debarshiray pushed a commit to yselkowitz/toolbox that referenced this pull request Sep 4, 2026
This allows using the 'distro' option to create and enter Fedora ELN
containers.  Due to ELN's rolling-release model, the 'release' option
isn't required.  If 'release' is used, the accepted value is 'latest'.

Note that the VERSION_ID field in os-release(5) on Fedora ELN contains a
number representing the next major version of Red Hat Enterprise Linux,
even though it follows a rolling-release model.  Right now it is:
  VERSION_ID=11

containers#1814

Signed-off-by: Yaakov Selkowitz <yselkowi@redhat.com>
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/fd050e68adb54733b40ca732d77757ea

✔️ unit-test SUCCESS in 2m 22s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 05s
✔️ unit-test-restricted SUCCESS in 2m 11s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 43m 08s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 32m 11s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 8m 44s
✔️ system-test-fedora-44-commands-options SUCCESS in 56m 38s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 31m 06s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 10m 15s
✔️ system-test-fedora-43-commands-options SUCCESS in 56m 57s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 30m 46s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 10m 43s
✔️ system-test-fedora-42-commands-options SUCCESS in 52m 21s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 43m 08s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 9m 07s
✔️ system-test-fedora-coreos-next-commands-options SUCCESS in 56m 29s
✔️ system-test-fedora-coreos-next-runtime-environment-arch-fedora SUCCESS in 49m 17s
✔️ system-test-fedora-coreos-next-runtime-environment-ubuntu SUCCESS in 13m 01s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 57m 22s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 34m 22s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 10m 15s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 42m 11s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 46m 05s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 10m 03s

debarshiray pushed a commit to yselkowitz/toolbox that referenced this pull request Sep 4, 2026
This allows using the 'distro' option to create and enter Fedora ELN
containers.  Due to ELN's rolling-release model, the 'release' option
isn't required.  If 'release' is used, the accepted value is 'latest'.

Note that the VERSION_ID field in os-release(5) on Fedora ELN contains a
number representing the next major version of Red Hat Enterprise Linux,
even though it follows a rolling-release model.  Right now it is:
  VERSION_ID=11

containers#1814

Signed-off-by: Yaakov Selkowitz <yselkowi@redhat.com>
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/835594af1f8a4c17b2c892c15ee01cb9

✔️ unit-test SUCCESS in 2m 07s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 01s
✔️ unit-test-restricted SUCCESS in 2m 09s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 1h 02m 08s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 50m 11s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 7m 14s
✔️ system-test-fedora-44-commands-options SUCCESS in 40m 35s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 49m 47s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 7m 40s
✔️ system-test-fedora-43-commands-options SUCCESS in 1h 00m 54s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 30m 42s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 10m 47s
✔️ system-test-fedora-42-commands-options SUCCESS in 33m 56s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 26m 29s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 9m 17s
✔️ system-test-fedora-coreos-next-commands-options SUCCESS in 40m 13s
✔️ system-test-fedora-coreos-next-runtime-environment-arch-fedora SUCCESS in 53m 55s
✔️ system-test-fedora-coreos-next-runtime-environment-ubuntu SUCCESS in 14m 31s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 1h 00m 15s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 54m 27s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 14m 34s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 1h 01m 11s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 54m 03s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 14m 24s

debarshiray pushed a commit to yselkowitz/toolbox that referenced this pull request Sep 5, 2026
This allows using the 'distro' option to create and enter Fedora ELN
containers.  Due to ELN's rolling-release model, the 'release' option
isn't required.  If 'release' is used, the accepted value is 'latest'.

Note that the VERSION_ID field in os-release(5) on Fedora ELN contains a
number representing the next major version of Red Hat Enterprise Linux,
even though it follows a rolling-release model.  Right now it is:
  VERSION_ID=11

Some changes by Debarshi Ray.

containers#1814

Signed-off-by: Yaakov Selkowitz <yselkowi@redhat.com>
@centosinfra-prod-github-app

Copy link
Copy Markdown

Merge Failed.

This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.

@debarshiray

Copy link
Copy Markdown
Member

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Merge Failed.

This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.

This allows using the 'distro' option to create and enter Fedora ELN
containers.  Due to ELN's rolling-release model, the 'release' option
isn't required.  If 'release' is used, the accepted value is 'latest'.

Note that the VERSION_ID field in os-release(5) on Fedora ELN contains a
number representing the next major version of Red Hat Enterprise Linux,
even though it follows a rolling-release model.  Right now it is:
  VERSION_ID=11

Some changes by Debarshi Ray.

containers#1814

Signed-off-by: Yaakov Selkowitz <yselkowi@redhat.com>
@centosinfra-prod-github-app

Copy link
Copy Markdown

Merge Failed.

This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.

@debarshiray

Copy link
Copy Markdown
Member

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/d041aeb912ca484aa97244ac00457efc

unit-test NODE_FAILURE Node(set) request 200-0000193282 failed in 0s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 04s
✔️ unit-test-restricted SUCCESS in 2m 12s
system-test-fedora-rawhide-commands-options NODE_FAILURE Node(set) request 200-0000193334 failed in 0s
system-test-fedora-rawhide-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193286 failed in 0s
system-test-fedora-rawhide-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193287 failed in 0s
system-test-fedora-44-commands-options NODE_FAILURE Node(set) request 200-0000193288 failed in 0s
system-test-fedora-44-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193289 failed in 0s
system-test-fedora-44-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193290 failed in 0s
system-test-fedora-43-commands-options NODE_FAILURE Node(set) request 200-0000193291 failed in 0s
system-test-fedora-43-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193292 failed in 0s
system-test-fedora-43-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193293 failed in 0s
system-test-fedora-42-commands-options NODE_FAILURE Node(set) request 200-0000193294 failed in 0s
system-test-fedora-42-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193295 failed in 0s
system-test-fedora-42-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193296 failed in 0s
system-test-fedora-coreos-next-commands-options NODE_FAILURE Node(set) request 200-0000193297 failed in 0s
system-test-fedora-coreos-next-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193298 failed in 0s
system-test-fedora-coreos-next-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193299 failed in 0s
system-test-fedora-coreos-stable-commands-options NODE_FAILURE Node(set) request 200-0000193300 failed in 0s
system-test-fedora-coreos-stable-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193301 failed in 0s
system-test-fedora-coreos-stable-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193302 failed in 0s
system-test-fedora-coreos-testing-commands-options NODE_FAILURE Node(set) request 200-0000193303 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193304 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193305 failed in 0s

@TristanCacqueray

Copy link
Copy Markdown
Contributor

recheck

@debarshiray

Copy link
Copy Markdown
Member

recheck

Thanks for poking at this on a Sunday! :)

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/41a167d89c3c49609410e8d15d58536c

unit-test NODE_FAILURE Node(set) request 200-0000193445 failed in 0s
unit-test-migration-path-for-coreos-toolbox NODE_FAILURE Node(set) request 200-0000193446 failed in 0s
unit-test-restricted NODE_FAILURE Node(set) request 200-0000193447 failed in 0s
system-test-fedora-rawhide-commands-options NODE_FAILURE Node(set) request 200-0000193448 failed in 0s
system-test-fedora-rawhide-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193449 failed in 0s
system-test-fedora-rawhide-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193450 failed in 0s
system-test-fedora-44-commands-options NODE_FAILURE Node(set) request 200-0000193451 failed in 0s
system-test-fedora-44-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193452 failed in 0s
system-test-fedora-44-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193453 failed in 0s
system-test-fedora-43-commands-options NODE_FAILURE Node(set) request 200-0000193454 failed in 0s
system-test-fedora-43-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193455 failed in 0s
system-test-fedora-43-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193456 failed in 0s
system-test-fedora-42-commands-options NODE_FAILURE Node(set) request 200-0000193457 failed in 0s
system-test-fedora-42-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193458 failed in 0s
system-test-fedora-42-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193459 failed in 0s
system-test-fedora-coreos-next-commands-options NODE_FAILURE Node(set) request 200-0000193460 failed in 0s
system-test-fedora-coreos-next-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193461 failed in 0s
system-test-fedora-coreos-next-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193462 failed in 0s
system-test-fedora-coreos-stable-commands-options NODE_FAILURE Node(set) request 200-0000193463 failed in 0s
system-test-fedora-coreos-stable-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193464 failed in 0s
system-test-fedora-coreos-stable-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193465 failed in 0s
system-test-fedora-coreos-testing-commands-options NODE_FAILURE Node(set) request 200-0000193466 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193467 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193468 failed in 0s

@TristanCacqueray

Copy link
Copy Markdown
Contributor

Looks like the provider is having some difficulties, we'll have a look tomorrow!

@debarshiray

Copy link
Copy Markdown
Member

recheck

@mhuin

mhuin commented Sep 7, 2026

Copy link
Copy Markdown

FYI there was a planned maintenance on vexxhost (our cloud provider) over this weekend, which probably resulted in the issues you faced. Now that the maintenance is over, you shouldn't see this type of issues anymore.

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/13cc5c1d43ea4f19821c373c177aa9be

✔️ unit-test SUCCESS in 2m 41s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 4m 00s
✔️ unit-test-restricted SUCCESS in 2m 41s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 1h 04m 20s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 45m 07s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 9m 17s
✔️ system-test-fedora-44-commands-options SUCCESS in 44m 46s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 36m 07s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 9m 34s
✔️ system-test-fedora-43-commands-options SUCCESS in 45m 50s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 36m 36s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 9m 43s
✔️ system-test-fedora-42-commands-options SUCCESS in 41m 29s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 32m 38s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 8m 18s
✔️ system-test-fedora-coreos-next-commands-options SUCCESS in 1h 04m 14s
✔️ system-test-fedora-coreos-next-runtime-environment-arch-fedora SUCCESS in 57m 24s
✔️ system-test-fedora-coreos-next-runtime-environment-ubuntu SUCCESS in 15m 04s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 1h 07m 16s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 57m 48s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 15m 38s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 1h 07m 11s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 58m 27s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 15m 41s

@debarshiray

Copy link
Copy Markdown
Member

FYI there was a planned maintenance on vexxhost (our cloud provider) over this weekend, which probably resulted in the issues you faced. Now that the maintenance is over, you shouldn't see this type of issues anymore.

I see! Yes, that makes sense. Thanks, @mhuin

@debarshiray
debarshiray merged commit ab122f5 into containers:main Sep 8, 2026
3 checks passed
@debarshiray

Copy link
Copy Markdown
Member

Thanks for the contribution, @yselkowitz !

@mhuin @Rolv-Apneseth @TristanCacqueray thanks for all the help with the CI!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants