Update websocket-client requirement from !=0.40.0,!=0.41.*,!=0.42.*,>=0.32.0 to !=0.40.0,!=0.41.0.dev,!=0.42.0.dev,>=1.9.2 - #2708
Conversation
Updates the requirements on [websocket-client](https://github.com/websocket-client/websocket-client) to permit the latest version. - [Release notes](https://github.com/websocket-client/websocket-client/releases) - [Changelog](https://github.com/websocket-client/websocket-client/blob/master/ChangeLog) - [Commits](websocket-client/websocket-client@v0.32.0...v1.9.2) --- updated-dependencies: - dependency-name: websocket-client dependency-version: 1.9.2 dependency-type: direct:production ... Signed-off-by: dependabot[bot] <support@github.com>
|
Adding the "do-not-merge/release-note-label-needed" label because no release-note block was detected, please follow our release note process to remove it. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
/close |
|
@yliaog: Closed this PR. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
OK, I won't notify you again about this release, but will get in touch when a new version is available. If you'd rather skip all updates until the next major or minor version, let me know by commenting If you change your mind, just re-open this PR and I'll resolve any conflicts on it. |
Updates the requirements on websocket-client to permit the latest version.
Release notes
Sourced from websocket-client's releases.
Changelog
Sourced from websocket-client's changelog.
... (truncated)
Commits
12225f51.9.209f2828Fix race in WebSocketApp.close() when run_forever thread clears sock (#1057)b5435d61.9.126f1c64Fix custom dispatcher SSL errors and socket timeouts (#1049)2a1b489Add ConnectionResetError to improve custom dispatcher error handling when ser...10d9159Improve close frame handling to fix #1043 and add tests (#1044)75a361aClosing open issues (#1037)e46a8fe1.9.0ab81eabBug fixes (#1036)63058abRun forever socket close fix (#1015)Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore this major versionwill close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this minor versionwill close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this dependencywill close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)