Skip to content

NO-ISSUE: rebase-release-4.22-4.22.0-0.nightly-2026-09-13-081542_amd64-2026-09-13_arm64-2026-09-14 - #7369

Open
microshift-rebase-script[bot] wants to merge 5 commits into
release-4.22from
rebase-release-4.22-4.22.0-0.nightly-2026-09-13-081542_amd64-2026-09-13_arm64-2026-09-14
Open

NO-ISSUE: rebase-release-4.22-4.22.0-0.nightly-2026-09-13-081542_amd64-2026-09-13_arm64-2026-09-14#7369
microshift-rebase-script[bot] wants to merge 5 commits into
release-4.22from
rebase-release-4.22-4.22.0-0.nightly-2026-09-13-081542_amd64-2026-09-13_arm64-2026-09-14

Conversation

@microshift-rebase-script

Copy link
Copy Markdown
Contributor

amd64: 4.22.0-0.nightly-2026-09-13-081542
arm64: 4.22.0-0.nightly-arm64-2026-09-14-021803
prow job: https://prow.ci.openshift.org/view/gs/origin-ci-test/logs/periodic-ci-openshift-microshift-release-4.22-periodics-rebase-on-nightlies/2099347769964630016

  • cluster-ingress-operator embedded-component 656d475a0bc76b31b8e387abd9e19458531cc4a3 to ab5f68632a64486c295c84301600f630c6693976

    • feb77cf 2026-08-26T13:14:55-04:00 vendor: updated google auth apis
    • cc12883 2026-08-26T13:14:55-04:00 go.mod: bump google api to v0.199.0
    • 864e8fb 2026-08-26T13:14:55-04:00 GCP: Set Universe Domain
  • cluster-kube-apiserver-operator embedded-component a2ba9c6a2d1ed76ea98a40885cf9a824eb9b9071 to 6269fbc7b6ae116ff8e7e22cd24ca047fab18e1f

  • cluster-kube-controller-manager-operator embedded-component c5c678b3e04a64214ae62f5ad6d873c8899ac363 to 99a72b4e43a0e019a15a53739873f2f868c05dd7

  • machine-config-operator embedded-component 90b269666abb8df37c46bf49d9b54dcf2604447c to f3334a057822a0f01ba82e15243c6ced44899a30

    • c2572ef1 2026-09-11T08:49:48Z OCPBUGS-69681: limit ContainerRuntimeConfig status condition to 3
  • service-ca-operator embedded-component 9fff46a576033685f66fbb1121eb0a2f134a29be to de05110fb2993491ff7f40455deb2f70cb95423b

    • fa94a82 2026-09-10T22:23:53Z OCPBUGS-114426: Update build-machinery-go vendor dependency
  • oc image-amd64 96ed0f43f0b5adf771ce3bd4fbc86809e93139f8 to aea5a82b408a8b83dd7d6f8b8ee9a83585dc87fd

    • 4ec6cbb6 2026-09-10T22:24:35Z OCPBUGS-114426: Update build-machinery-go vendor dependency
  • service-ca-operator image-amd64 9fff46a576033685f66fbb1121eb0a2f134a29be to de05110fb2993491ff7f40455deb2f70cb95423b

    • fa94a82 2026-09-10T22:23:53Z OCPBUGS-114426: Update build-machinery-go vendor dependency
  • oc image-arm64 96ed0f43f0b5adf771ce3bd4fbc86809e93139f8 to aea5a82b408a8b83dd7d6f8b8ee9a83585dc87fd

    • 4ec6cbb6 2026-09-10T22:24:35Z OCPBUGS-114426: Update build-machinery-go vendor dependency
  • service-ca-operator image-arm64 9fff46a576033685f66fbb1121eb0a2f134a29be to de05110fb2993491ff7f40455deb2f70cb95423b

    • fa94a82 2026-09-10T22:23:53Z OCPBUGS-114426: Update build-machinery-go vendor dependency

/label tide/merge-method-squash
/label backport-risk-assessed
/label jira/valid-bug

@microshift-rebase-script

Copy link
Copy Markdown
Contributor Author

/pipeline auto

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 14, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@microshift-rebase-script[bot]: This pull request explicitly references no jira issue.

Details

In response to this:

amd64: 4.22.0-0.nightly-2026-09-13-081542
arm64: 4.22.0-0.nightly-arm64-2026-09-14-021803
prow job: https://prow.ci.openshift.org/view/gs/origin-ci-test/logs/periodic-ci-openshift-microshift-release-4.22-periodics-rebase-on-nightlies/2099347769964630016

  • cluster-ingress-operator embedded-component 656d475a0bc76b31b8e387abd9e19458531cc4a3 to ab5f68632a64486c295c84301600f630c6693976

  • feb77cf 2026-08-26T13:14:55-04:00 vendor: updated google auth apis

  • cc12883 2026-08-26T13:14:55-04:00 go.mod: bump google api to v0.199.0

  • 864e8fb 2026-08-26T13:14:55-04:00 GCP: Set Universe Domain

  • cluster-kube-apiserver-operator embedded-component a2ba9c6a2d1ed76ea98a40885cf9a824eb9b9071 to 6269fbc7b6ae116ff8e7e22cd24ca047fab18e1f

  • 8def64b 2026-03-17T06:13:45Z Updating ose-cluster-kube-apiserver-operator-container image to be consistent with ART for 4.22 Reconciling with https://github.com/openshift/ocp-build-data/tree/0daf002c3db04b759c8f818f28800c1531cf2f79/images/ose-cluster-kube-apiserver-operator.yml

  • cluster-kube-controller-manager-operator embedded-component c5c678b3e04a64214ae62f5ad6d873c8899ac363 to 99a72b4e43a0e019a15a53739873f2f868c05dd7

  • 9c807e1 2026-03-03T16:52:38Z Updating ose-cluster-kube-controller-manager-operator-container image to be consistent with ART for 4.22 Reconciling with https://github.com/openshift/ocp-build-data/tree/56cb39ad358cdec1db7c84ea1919fe8849c2550b/images/ose-cluster-kube-controller-manager-operator.yml

  • machine-config-operator embedded-component 90b269666abb8df37c46bf49d9b54dcf2604447c to f3334a057822a0f01ba82e15243c6ced44899a30

  • c2572ef1 2026-09-11T08:49:48Z OCPBUGS-69681: limit ContainerRuntimeConfig status condition to 3

  • service-ca-operator embedded-component 9fff46a576033685f66fbb1121eb0a2f134a29be to de05110fb2993491ff7f40455deb2f70cb95423b

  • fa94a82 2026-09-10T22:23:53Z OCPBUGS-114426: Update build-machinery-go vendor dependency

  • oc image-amd64 96ed0f43f0b5adf771ce3bd4fbc86809e93139f8 to aea5a82b408a8b83dd7d6f8b8ee9a83585dc87fd

  • 4ec6cbb6 2026-09-10T22:24:35Z OCPBUGS-114426: Update build-machinery-go vendor dependency

  • service-ca-operator image-amd64 9fff46a576033685f66fbb1121eb0a2f134a29be to de05110fb2993491ff7f40455deb2f70cb95423b

  • fa94a82 2026-09-10T22:23:53Z OCPBUGS-114426: Update build-machinery-go vendor dependency

  • oc image-arm64 96ed0f43f0b5adf771ce3bd4fbc86809e93139f8 to aea5a82b408a8b83dd7d6f8b8ee9a83585dc87fd

  • 4ec6cbb6 2026-09-10T22:24:35Z OCPBUGS-114426: Update build-machinery-go vendor dependency

  • service-ca-operator image-arm64 9fff46a576033685f66fbb1121eb0a2f134a29be to de05110fb2993491ff7f40455deb2f70cb95423b

  • fa94a82 2026-09-10T22:23:53Z OCPBUGS-114426: Update build-machinery-go vendor dependency

/label tide/merge-method-squash
/label backport-risk-assessed
/label jira/valid-bug

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@coderabbitai

coderabbitai Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Important

Review skipped

Bot user detected.

To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration

Configuration used: Repository YAML (base), Central YAML (inherited)

Review profile: CHILL

Plan: Advanced

Run ID: 87e7a6bf-4a0e-49f8-a6c9-9f2a40fc14af

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Comment @coderabbitai help to get the list of available commands.

@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@openshift-ci openshift-ci Bot added tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. pipeline-auto labels Sep 14, 2026
@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Pipeline controller notification

The pipeline-auto label has been added to this PR. Second-stage tests will be triggered automatically when all first-stage tests pass.

@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: microshift-rebase-script[bot]
Once this PR has been reviewed and has the lgtm label, please assign copejon for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 7660566 and 2 for PR HEAD 58281d3 in total

@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD b55dfb4 and 1 for PR HEAD 58281d3 in total

@microshift-rebase-script

Copy link
Copy Markdown
Contributor Author

/test images
/test ocp-full-conformance-rhel-eus
/test test-unit

Added by prow-jobs-for-pull-requests.sh 🤖

@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

@microshift-rebase-script[bot]: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Scheduling tests matching the pipeline_run_if_changed or not excluded by pipeline_skip_if_only_changed parameters:
/test e2e-aws-tests
/test e2e-aws-tests-arm
/test e2e-aws-tests-bootc-arm-el10
/test e2e-aws-tests-bootc-arm-el9
/test e2e-aws-tests-bootc-el10
/test e2e-aws-tests-bootc-el9

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. pipeline-auto tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant